chore(deps): update all digest updates #127

Merged
renovate[bot] merged 1 commits from renovate/all-digest into master 2026-09-13 04:02:04 +04:00
Collaborator

This PR contains the following updates:

Package Change Age Confidence
autoprefixer 10.5.610.6.0 age confidence
sass 1.104.01.104.1 age confidence

Release Notes

postcss/autoprefixer (autoprefixer)

v10.6.0

Compare Source

sass/dart-sass (sass)

v1.104.1

Compare Source

  • Fix a bug where loud comments before @use rules could be emitted multiple
    times under certain circumstances.

  • Fix a bug in which an invalid selector at the end of an indented syntax
    stylesheet would cause the parser to crash instead of emitting a useful error
    message.

Command Line Interface
  • Many-to-many compilations no longer compile any files that appear in the
    output directory if the output directory is also within the source directory.
    This fixes a bug where --watch mode could enter an infinite loop recompiling
    the same CSS file over and over.

  • Sass now sets the modification time of output files to the time compilation
    started rather than the time it ended. This ensures that, if a source file
    is modified during compilation, --watch and --update mode will recompile
    the outputs to include the new source file contents.


Configuration

📅 Schedule: (UTC)

  • Branch creation
    • At any time (no schedule defined)
  • Automerge
    • At any time (no schedule defined)

🚦 Automerge: Enabled.

Rebasing: Whenever PR is behind base branch, or you tick the rebase/retry checkbox.

👻 Immortal: This PR will be recreated if closed unmerged. Get config help if that's undesired.


  • If you want to rebase/retry this PR, check this box

This PR has been generated by Mend Renovate CLI.

This PR contains the following updates: | Package | Change | [Age](https://docs.renovatebot.com/merge-confidence/) | [Confidence](https://docs.renovatebot.com/merge-confidence/) | |---|---|---|---| | [autoprefixer](https://github.com/postcss/autoprefixer) | [`10.5.6` → `10.6.0`](https://renovatebot.com/diffs/npm/autoprefixer/10.5.6/10.6.0) | ![age](https://developer.mend.io/api/mc/badges/age/npm/autoprefixer/10.6.0?slim=true) | ![confidence](https://developer.mend.io/api/mc/badges/confidence/npm/autoprefixer/10.5.6/10.6.0?slim=true) | | [sass](https://github.com/sass/dart-sass) | [`1.104.0` → `1.104.1`](https://renovatebot.com/diffs/npm/sass/1.104.0/1.104.1) | ![age](https://developer.mend.io/api/mc/badges/age/npm/sass/1.104.1?slim=true) | ![confidence](https://developer.mend.io/api/mc/badges/confidence/npm/sass/1.104.0/1.104.1?slim=true) | --- ### Release Notes <details> <summary>postcss/autoprefixer (autoprefixer)</summary> ### [`v10.6.0`](https://github.com/postcss/autoprefixer/blob/HEAD/CHANGELOG.md#1060-Post-tenebras-lux) [Compare Source](https://github.com/postcss/autoprefixer/compare/10.5.6...10.6.0) - Added `initial-letter` support (by [@&#8203;thomasmattheussen](https://github.com/thomasmattheussen)). </details> <details> <summary>sass/dart-sass (sass)</summary> ### [`v1.104.1`](https://github.com/sass/dart-sass/blob/HEAD/CHANGELOG.md#11041) [Compare Source](https://github.com/sass/dart-sass/compare/1.104.0...1.104.1) - Fix a bug where loud comments before `@use` rules could be emitted multiple times under certain circumstances. - Fix a bug in which an invalid selector at the end of an indented syntax stylesheet would cause the parser to crash instead of emitting a useful error message. ##### Command Line Interface - Many-to-many compilations no longer compile any files that appear in the output directory if the output directory is also within the source directory. This fixes a bug where `--watch` mode could enter an infinite loop recompiling the same CSS file over and over. - Sass now sets the modification time of output files to the time compilation *started* rather than the time it *ended*. This ensures that, if a source file is modified during compilation, `--watch` and `--update` mode will recompile the outputs to include the new source file contents. </details> --- ### Configuration 📅 **Schedule**: (UTC) - Branch creation - At any time (no schedule defined) - Automerge - At any time (no schedule defined) 🚦 **Automerge**: Enabled. ♻ **Rebasing**: Whenever PR is behind base branch, or you tick the rebase/retry checkbox. 👻 **Immortal**: This PR will be recreated if closed unmerged. Get [config help](https://github.com/renovatebot/renovate/discussions) if that's undesired. --- - [ ] <!-- rebase-check -->If you want to rebase/retry this PR, check this box --- This PR has been generated by [Mend Renovate CLI](https://github.com/renovatebot/renovate). <!--renovate-debug:eyJjcmVhdGVkSW5WZXIiOiI0NC44Mi40IiwidXBkYXRlZEluVmVyIjoiNDQuODIuNCIsInRhcmdldEJyYW5jaCI6Im1hc3RlciIsImxhYmVscyI6WyJhdXRvbWF0ZWQiLCJkZXBlbmRlbmNpZXMiXX0=-->
renovate[bot] added 1 commit 2026-09-13 04:02:02 +04:00
chore(deps): update all digest updates
Test / npm test (pull_request) Successful in 38s
RenovateBot / renovate (push) Successful in 30s
Test / npm test (push) Successful in 38s
2017d662dd
renovate[bot] scheduled this pull request to auto merge when all checks succeed 2026-09-13 04:02:02 +04:00
renovate[bot] merged commit 2017d662dd into master 2026-09-13 04:02:04 +04:00
renovate[bot] deleted branch renovate/all-digest 2026-09-13 04:02:04 +04:00
Sign in to join this conversation.
No Reviewers
No labels
1 Participants
Notifications
Due Date
No due date set.
Dependencies

No dependencies set.

Reference: valentineus/popov.link#127